Skip to content

[stable34] chore: Update CODEOWNERS for IDP team - #65285

Merged
susnux merged 1 commit into
stable34from
backport/65277/stable34
Oct 7, 2026
Merged

susnux merged 1 commit into
stable34from
backport/65277/stable34

Conversation

@backportbot

@backportbot backportbot Bot commented Oct 7, 2026

Copy link
Copy Markdown

Backport of PR #65277

Signed-off-by: Carl Schwan <carl@carlschwan.eu>
@backportbot
backportbot Bot requested a review from a team as a code owner October 7, 2026 15:34
@backportbot
backportbot Bot requested review from Altahrim, CarlSchwan, artonge, leftybournes, provokateurin, salmart-dev and susnux and removed request for a team October 7, 2026 15:34
@backportbot backportbot Bot added the 3. to review Waiting for reviews label Oct 7, 2026
@backportbot backportbot Bot added this to the Nextcloud 34.0.5 milestone Oct 7, 2026
@susnux
susnux merged commit 0c128f8 into stable34 Oct 7, 2026
93 of 96 checks passed
@susnux
susnux deleted the backport/65277/stable34 branch October 7, 2026 16:08
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

3. to review Waiting for reviews

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants