Skip to content

fix: iOS improvements, UI fixes, and pacakge updates - #269

Merged
WhyAsh5114 merged 3 commits into
mainfrom
fix/ios-improvements
Aug 29, 2026
Merged

fix: iOS improvements, UI fixes, and pacakge updates#269
WhyAsh5114 merged 3 commits into
mainfrom
fix/ios-improvements

Conversation

@WhyAsh5114

Copy link
Copy Markdown
Owner

Description

iOS improvements, UI fixes, and pacakge updates

Type of change

  • Bug fix (non-breaking change which fixes an issue)

@vercel

vercel Bot commented Aug 29, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
my-fit Ignored Ignored Aug 29, 2026 7:33am

@WhyAsh5114
WhyAsh5114 merged commit 20ed13b into main Aug 29, 2026
6 checks passed
@WhyAsh5114
WhyAsh5114 deleted the fix/ios-improvements branch August 29, 2026 07:37
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant