Skip to content

DREF Landing: Update content - #2530

Closed
frozenhelium wants to merge 1 commit into
developfrom
feat/dref-landing-page-feedbacks
Closed

frozenhelium wants to merge 1 commit into
developfrom
feat/dref-landing-page-feedbacks

Conversation

@frozenhelium

Copy link
Copy Markdown
Member

No description provided.

@changeset-bot

changeset-bot Bot commented Sep 2, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 40bce2a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
go-web-app Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@frozenhelium
frozenhelium force-pushed the feat/dref-landing-page-feedbacks branch from 59908e1 to cad4fe3 Compare September 2, 2026 07:06
@frozenhelium
frozenhelium force-pushed the feat/dref-landing-page-feedbacks branch from cad4fe3 to 40bce2a Compare September 2, 2026 07:49
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant